feature: update deps and buf - #62
Conversation
Signed-off-by: Valery Piashchynski <piashchynski.valery@gmail.com>
Signed-off-by: Valery Piashchynski <piashchynski.valery@gmail.com>
|
Warning Review failedThe pull request is closed. WalkthroughThese updates collectively advance the Changes
Poem
Thank you for using CodeRabbit. We offer it for free to the OSS community and would appreciate your support in helping us grow. If you find it useful, would you consider giving us a shout-out on your favorite social media? TipsChatThere are 3 ways to chat with CodeRabbit:
Note: Be mindful of the bot's finite context window. It's strongly recommended to break down tasks such as reading entire modules into smaller chunks. For a focused discussion, use review comments to chat about specific files and their changes, instead of using the PR comments. CodeRabbit Commands (invoked as PR comments)
Additionally, you can add CodeRabbit Configration File (
|
There was a problem hiding this comment.
Actionable comments posted: 0
Review details
Configuration used: CodeRabbit UI
Review profile: CHILL
Files ignored due to path filters (65)
buf.lockis excluded by!**/*.lockbuild/applogger/v1/applogger.pb.gois excluded by!**/*.pb.gobuild/centrifugo/api/v1/api.pb.gois excluded by!**/*.pb.gobuild/centrifugo/api/v1/api_grpc.pb.gois excluded by!**/*.pb.gobuild/centrifugo/proxy/v1/proxy.pb.gois excluded by!**/*.pb.gobuild/centrifugo/proxy/v1/proxy_grpc.pb.gois excluded by!**/*.pb.gobuild/common/v1/grpc_status.pb.gois excluded by!**/*.pb.gobuild/google/api/annotations.pb.gois excluded by!**/*.pb.gobuild/google/api/http.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/any.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/descriptor.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/duration.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/empty.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/struct.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/timestamp.pb.gois excluded by!**/*.pb.gobuild/google/protobuf/wrappers.pb.gois excluded by!**/*.pb.gobuild/http/v1/http.pb.gois excluded by!**/*.pb.gobuild/jobs/v1/jobs.pb.gois excluded by!**/*.pb.gobuild/kv/v1/kv.pb.gois excluded by!**/*.pb.gobuild/lock/v1beta1/lock.pb.gois excluded by!**/*.pb.gobuild/service/v1/service.pb.gois excluded by!**/*.pb.gobuild/status/v1/status.pb.gois excluded by!**/*.pb.gobuild/status/v1beta1/status.pb.gois excluded by!**/*.pb.gobuild/temporal/api/batch/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/command/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/common/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/batch_operation.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/command_type.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/common.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/event_type.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/failed_cause.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/namespace.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/query.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/reset.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/schedule.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/task_queue.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/update.pb.gois excluded by!**/*.pb.gobuild/temporal/api/enums/v1/workflow.pb.gois excluded by!**/*.pb.gobuild/temporal/api/errordetails/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/export/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/failure/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/filter/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/history/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/namespace/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/nexus/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/operatorservice/v1/request_response.pb.gois excluded by!**/*.pb.gobuild/temporal/api/operatorservice/v1/service.pb.gois excluded by!**/*.pb.gobuild/temporal/api/operatorservice/v1/service_grpc.pb.gois excluded by!**/*.pb.gobuild/temporal/api/protocol/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/query/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/replication/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/schedule/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/sdk/v1/task_complete_metadata.pb.gois excluded by!**/*.pb.gobuild/temporal/api/sdk/v1/user_metadata.pb.gois excluded by!**/*.pb.gobuild/temporal/api/sdk/v1/workflow_metadata.pb.gois excluded by!**/*.pb.gobuild/temporal/api/taskqueue/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/update/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/version/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/workflow/v1/message.pb.gois excluded by!**/*.pb.gobuild/temporal/api/workflowservice/v1/request_response.pb.gois excluded by!**/*.pb.gobuild/temporal/api/workflowservice/v1/service.pb.gois excluded by!**/*.pb.gobuild/temporal/api/workflowservice/v1/service_grpc.pb.gois excluded by!**/*.pb.gobuild/temporal/v1/temporal.pb.gois excluded by!**/*.pb.gobuild/websockets/v1/websockets.pb.gois excluded by!**/*.pb.gogo.sumis excluded by!**/*.sum
Files selected for processing (19)
- buf.gen.yaml (1 hunks)
- buf.yaml (1 hunks)
- go.mod (1 hunks)
- plugins/v4/jobs/driver.go (1 hunks)
- plugins/v4/jobs/job.go (1 hunks)
- plugins/v4/jobs/state.go (1 hunks)
- plugins/v4/priority_queue/priority_queue.go (1 hunks)
- proto/applogger/v1/applogger.proto (1 hunks)
- proto/common/v1/grpc_status.proto (1 hunks)
- proto/http/v1/http.proto (1 hunks)
- proto/jobs/v1/jobs.proto (1 hunks)
- proto/kv/v1/kv.proto (1 hunks)
- proto/lock/v1beta1/lock.proto (1 hunks)
- proto/service/v1/service.proto (1 hunks)
- proto/status/v1/status.proto (1 hunks)
- proto/status/v1beta1/status.proto (1 hunks)
- proto/temporal/v1/temporal.proto (1 hunks)
- proto/websockets/v1/websockets.proto (1 hunks)
- third_party/api (1 hunks)
Files skipped from review due to trivial changes (9)
- buf.yaml
- plugins/v4/priority_queue/priority_queue.go
- proto/applogger/v1/applogger.proto
- proto/lock/v1beta1/lock.proto
- proto/service/v1/service.proto
- proto/status/v1beta1/status.proto
- proto/temporal/v1/temporal.proto
- proto/websockets/v1/websockets.proto
- third_party/api
Additional comments not posted (18)
buf.gen.yaml (3)
1-1: Version update approved.The upgrade to
version: v2is consistent with the PR objectives and the AI-generated summary.
3-5: Check the plugin version and options.The update to
buf.build/protocolbuffers/go:v1.34.2without: buildandopt: paths=source_relativealigns with the latest standards and should facilitate better source management.
6-10: Review new plugin configuration.The
buf.build/grpc/go:v1.4.0plugin configuration includes options for paths and server requirements. This is a positive change, ensuring better compatibility and control over generated code.proto/status/v1/status.proto (1)
5-5: Updated Go package path approved.The change in
go_packageto"github.com/roadrunner-server/api/v4/build/status/v1;statusV1"reflects the new project structure and should help in maintaining clear package management.proto/kv/v1/kv.proto (1)
5-5: Updated Go package path approved.The update to
go_packageto"github.com/roadrunner-server/api/v4/build/kv/v1;kvV1"is consistent with the restructuring of the project's directories and namespaces.go.mod (3)
3-3: Go version update approved.Upgrading the Go version to
1.22.4should provide the latest features and security patches.
6-10: Dependencies update approved.The updates to
go.temporal.io/api,google.golang.org/grpc, andgoogle.golang.org/protobufare crucial for keeping the project up-to-date with the latest improvements and security features.
15-18: Indirect dependencies update approved.Updating indirect dependencies such as
golang.org/x/netandgoogle.golang.org/genproto/googleapis/rpcensures compatibility and the latest patches are applied.proto/http/v1/http.proto (1)
7-7: Updated Go package path.The
go_packageoption has been updated to reflect the new package structure. Please ensure that the new path is correctly integrated and recognized in the Go project.proto/common/v1/grpc_status.proto (1)
7-7: Updated Go package path.The
go_packageoption has been updated to align with the new package structure. Please verify that this change does not introduce any issues with protobuf compilation or package resolution.Verification successful
Updated Go package path verified.
The
go_packageoption has been correctly updated to the new package structure, and no references to the old package path were found.
proto/common/v1/grpc_status.proto: The newgo_packagepath is consistent.- No remaining references to the old package path detected.
Scripts executed
The following scripts were executed for the analysis:
Script:
#!/bin/bash # Description: Verify the new Go package path is correctly recognized. # Test: Compile the protobuf files. Expect: No errors. protoc --proto_path=./ --go_out=./ ./proto/common/v1/grpc_status.protoLength of output: 146
Script:
#!/bin/bash # Check for consistency of the new go_package path echo "Checking for new go_package path consistency..." rg 'github.com/roadrunner-server/api/v4/build/common/v1;commonV1' # Check for any remaining references to the old go_package path echo "Checking for old go_package path references..." rg 'github.com/roadrunner-server/api/v4/common/v1;commonV1'Length of output: 455
plugins/v4/jobs/state.go (1)
1-40: Introduced constants, Command type, and State struct.The additions include well-defined constants for job message packing, a new
Commandtype, and aStatestruct representing various job states. Please ensure these new types and constants are used consistently and correctly across the application.plugins/v4/jobs/driver.go (1)
1-39: Introduced Driver, Commander, and Constructor interfaces.These interfaces provide structured methods for job management. Ensure that they are implemented correctly and integrated smoothly with the rest of the application.
proto/jobs/v1/jobs.proto (1)
5-5: Updatedgo_packagepath approved.The change in the
go_packagepath aligns with standard practices for managing Go packages, especially in larger projects that require structured versioning and modularization.plugins/v4/jobs/job.go (5)
7-17:Queueinterface structure is appropriate for job management.The methods defined in the
Queueinterface (Remove,Insert,ExtractMin,Len) are standard and well-suited for managing a job queue within the Jobs plugin.
19-36: Addition ofNackWithOptionsmethod enhances job handling flexibility.The
Jobinterface is well-structured and the addition ofNackWithOptionsmethod is a valuable enhancement, providing more options for handling job failures, such as requeuing with a delay.
38-55:Messageinterface effectively encapsulates job metadata and operations.The structure of the
Messageinterface, integratingKafkaOptions, is well-designed for handling job-related data, especially in systems utilizing Kafka. The methods provide essential job metadata and operations, making it robust for various job handling scenarios.
57-67:KafkaOptionsinterface is appropriately designed for Kafka-specific job properties.The methods in the
KafkaOptionsinterface (Offset,Partition,Topic,Metadata) are specific and relevant for managing Kafka-related job properties. This design ensures that Kafka-specific functionalities are cleanly abstracted and managed.
69-91:Pipelineinterface provides comprehensive methods for pipeline configuration and management.The
Pipelineinterface, with its methods for setting, retrieving, and checking pipeline-specific settings, is crucial for the flexible and effective management of job pipelines. This interface is well-designed to handle various configurations and metadata associated with job pipelines.
Signed-off-by: Valery Piashchynski <piashchynski.valery@gmail.com>
There was a problem hiding this comment.
Actionable comments posted: 0
Review details
Configuration used: CodeRabbit UI
Review profile: CHILL
Files selected for processing (1)
- plugins/v4/jobs/job.go (1 hunks)
Files skipped from review as they are similar to previous changes (1)
- plugins/v4/jobs/job.go
Signed-off-by: Valery Piashchynski <piashchynski.valery@gmail.com>
There was a problem hiding this comment.
Actionable comments posted: 0
Review details
Configuration used: CodeRabbit UI
Review profile: CHILL
Files selected for processing (1)
- plugins/v4/jobs/job.go (1 hunks)
Files skipped from review as they are similar to previous changes (1)
- plugins/v4/jobs/job.go
Signed-off-by: Valery Piashchynski <piashchynski.valery@gmail.com>
There was a problem hiding this comment.
Actionable comments posted: 0
Review details
Configuration used: CodeRabbit UI
Review profile: CHILL
Files selected for processing (1)
- plugins/v4/jobs/job.go (1 hunks)
Files skipped from review as they are similar to previous changes (1)
- plugins/v4/jobs/job.go
Signed-off-by: Valery Piashchynski <piashchynski.valery@gmail.com>
Reason for This PR
ref: roadrunner-php/jobs#66
Description of Changes
License Acceptance
By submitting this pull request, I confirm that my contribution is made under
the terms of the MIT license.
PR Checklist
[Author TODO: Meet these criteria.][Reviewer TODO: Verify that these criteria are met. Request changes if not]git commit -s).CHANGELOG.md.Summary by CodeRabbit
New Features
Bug Fixes
go_packageoptions in various.protofiles for improved Go code generation alignment.Chores
go.mod.buf.gen.yamlandbuf.yamlto use versionv2.Refactor